Skip to content

chore: bump manual-approval-gate from v0.8.0 to v0.8.1 - #4077

Merged
tekton-robot merged 1 commit into
release-v0.79.xfrom
bot-bump-payload-release-v0.79.x
Sep 11, 2026
Merged

chore: bump manual-approval-gate from v0.8.0 to v0.8.1#4077
tekton-robot merged 1 commit into
release-v0.79.xfrom
bot-bump-payload-release-v0.79.x

Conversation

@tekton-robot

Copy link
Copy Markdown
Contributor

Component Version Updates

This PR updates the following component versions:

manual-approval-gate


This PR was automatically created by the bump-payload workflow.

- manual-approval-gate: v0.8.0 → v0.8.1

Signed-off-by: tekton-bot <tekton-bot@users.noreply.github.com>
@tekton-robot tekton-robot added do-not-merge/release-note-label-needed Indicates that a PR should not merge because it's missing one of the release note labels. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 11, 2026
@divyansh42

Copy link
Copy Markdown
Member

/release-note-none

@tekton-robot tekton-robot added release-note-none Denotes a PR that doesnt merit a release note. and removed do-not-merge/release-note-label-needed Indicates that a PR should not merge because it's missing one of the release note labels. labels Sep 11, 2026
@pratap0007

Copy link
Copy Markdown
Contributor

/retest

@pratap0007

Copy link
Copy Markdown
Contributor

@divyansh42 The release manifest files are missing from MAG 0.8.1, which is causing the CI to fail. Could you please check the MAG 0.8.1 release?

@divyansh42

Copy link
Copy Markdown
Member

/retest

1 similar comment
@pratap0007

Copy link
Copy Markdown
Contributor

/retest

@pratap0007

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 11, 2026
@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 11, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: pratap0007, savitaashture, theakshaypant

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot
tekton-robot merged commit bbcbb29 into release-v0.79.x Sep 11, 2026
26 of 39 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesnt merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants