Skip to content

chore: bump component versions - #4076

Closed
tekton-robot wants to merge 1 commit into
release-v0.81.xfrom
bot-bump-payload
Closed

chore: bump component versions#4076
tekton-robot wants to merge 1 commit into
release-v0.81.xfrom
bot-bump-payload

Conversation

@tekton-robot

Copy link
Copy Markdown
Contributor

Component Version Updates

This PR updates the following component versions:

manual-approval-gate

pipeline

pipelines-as-code

chains

dashboard


This PR was automatically created by the bump-payload workflow.

- manual-approval-gate: v0.9.0 → v0.9.1
- pipeline: v1.15.1 → v1.16.0
- pipelines-as-code: v0.50.0 → v0.51.0
- chains: v0.29.3 → v0.29.4
- dashboard: v0.71.0 → v0.72.0

Signed-off-by: tekton-bot <tekton-bot@users.noreply.github.com>
@tekton-robot tekton-robot added the release-note-none Denotes a PR that doesnt merit a release note. label Sep 11, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To complete the pull request process, please assign pratap0007 after the PR has been reviewed.
You can assign the PR to them by writing /assign @pratap0007 in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 11, 2026
@divyansh42 divyansh42 closed this Sep 11, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

release-note-none Denotes a PR that doesnt merit a release note. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants