Skip to content

chore: bump payload versions - #4058

Merged
tekton-robot merged 1 commit into
mainfrom
bot-bump-payload
Sep 9, 2026
Merged

chore: bump payload versions#4058
tekton-robot merged 1 commit into
mainfrom
bot-bump-payload

Conversation

@tekton-robot

Copy link
Copy Markdown
Contributor

Automated component version bump

Signed-off-by: tekton-bot <tekton-bot@users.noreply.github.com>
@tekton-robot tekton-robot added the release-note-none Denotes a PR that doesnt merit a release note. label Sep 9, 2026
@tekton-robot
tekton-robot requested review from enarha and khrm September 9, 2026 01:26
@tekton-robot tekton-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 9, 2026
@codecov

codecov Bot commented Sep 9, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 26.70%. Comparing base (9df99c4) to head (d010137).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #4058   +/-   ##
=======================================
  Coverage   26.70%   26.70%           
=======================================
  Files         465      465           
  Lines       25010    25010           
=======================================
  Hits         6679     6679           
  Misses      17601    17601           
  Partials      730      730           
Flag Coverage Δ
unit-tests 26.70% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@jkhelil

jkhelil commented Sep 9, 2026

Copy link
Copy Markdown
Member

/approve

@jkhelil

jkhelil commented Sep 9, 2026

Copy link
Copy Markdown
Member

/lgtm

@tekton-robot

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jkhelil

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. labels Sep 9, 2026
@tekton-robot
tekton-robot merged commit 6a12f17 into main Sep 9, 2026
17 of 18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesnt merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants