chore: bump payload versions - #4055
Open
tekton-robot wants to merge 1 commit into
Open
Conversation
Contributor
|
/retest |
Contributor
|
/lgtm |
Contributor
|
/ok-to-test |
enarha
reviewed
Sep 8, 2026
enarha
left a comment
Contributor
There was a problem hiding this comment.
I see the same version in 2 other files. Shouldn't we update them all?
git grep operator-sdk-v1.7.2
operatorhub/kubernetes/manifests/bases/tektoncd-operator.clusterserviceversion.template.yaml: operators.operatorframework.io/builder: operator-sdk-v1.7.2
operatorhub/kubernetes/manifests/bases/tektoncd-operator.clusterserviceversion.template.yaml.tmpl: operators.operatorframework.io/builder: operator-sdk-v1.7.2
operatorhub/openshift/manifests/bases/openshift-pipelines-operator-rh.clusterserviceversion.template.yaml: operators.operatorframework.io/builder: operator-sdk-v1.7.2
operatorhub/openshift/manifests/bases/openshift-pipelines-operator-rh.clusterserviceversion.template.yaml.tmpl: operators.operatorframework.io/builder: operator-sdk-v1.7.2
anithapriyanatarajan
force-pushed
the
bot-bump-payload-release-v0.81.x
branch
from
September 8, 2026 09:55
3efee5e to
8bda00d
Compare
Contributor
Author
|
New changes are detected. LGTM label has been removed. |
Contributor
Author
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Signed-off-by: tekton-bot <tekton-bot@users.noreply.github.com>
tekton-robot
force-pushed
the
bot-bump-payload-release-v0.81.x
branch
from
September 10, 2026 02:06
8bda00d to
fa1b017
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Automated component version bump for release-v0.81.x