OCPBUGS-126704: Restart pods when rendering egressIP settings - #3168
raphaelvrosa wants to merge 1 commit into
Conversation
Issue: EgressIP ReachabilityTotalTimeoutSeconds is rendered by ovnkube-config (configmap), but ovnkube node and control pods are not restarted to have the new value applied in their settings. Solution: Make sure the spec templates of ovnkube node and control-plane pods take into account the changes in the ReachabilityTotalTimeoutSeconds setting. This is implemented by rendering the cli flags in their spec template command. When this value is changed, it then triggers the restart of those pods, reloading the parameter in their configuration. Adds test cases to evaluate the rendering and the hash changes in ovnkube node and control pod specs. Updates README to document that reachabilityTimeoutSeconds is enabled for runtime updates. Signed-off-by: Raphael Rosa <raprosa@redhat.com> (cherry picked from commit 88f6a1a)
|
Pipeline controller notification For optional jobs, comment This repository is configured in: LGTM mode |
|
@raphaelvrosa: This pull request explicitly references no jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
/pipeline required @kyrtapz can you take a look? |
|
Scheduling tests matching the |
|
/jira cherry-pick OCPBUGS-99048 |
|
@raphaelvrosa: Jira Issue OCPBUGS-99048 has been cloned as Jira Issue OCPBUGS-126704. Will retitle bug to link to clone. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@raphaelvrosa: This pull request references Jira Issue OCPBUGS-126704, which is valid. The bug has been moved to the POST state. 7 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (core-networking-bot@redhat.com), skipping review request. The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@raphaelvrosa once a hypershift job completes please go through the artifacts to verify that the flag is being set. |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: kyrtapz, raphaelvrosa The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/label backport-risk-assessed |
|
/retest required |
|
/retest-required |
@kyrtapz the default values are set in ovnkube-node and ovn-cluster-manager pods. Any specific tests to look into hypershift? |
|
@raphaelvrosa: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
Issue: EgressIP ReachabilityTotalTimeoutSeconds is rendered by ovnkube-config (configmap), but ovnkube node and control pods are not restarted to have the new value applied in their settings.
Solution: Make sure the spec templates of ovnkube node and control-plane pods take into account the changes in the ReachabilityTotalTimeoutSeconds setting. This is implemented by rendering the cli flags in their spec template command. When this value is changed, it then triggers the restart of those pods, reloading the parameter in their configuration.
Adds test cases to evaluate the rendering and the hash changes in ovnkube node and control pod specs.
Updates README to document that reachabilityTimeoutSeconds is enabled for runtime updates.
(cherry picked from commit 88f6a1a)
Conflicts addressed: