Skip to content

chore(deps): update opc-1-20-opc to 64e3b54 - #29422

Merged
pramodbindal merged 1 commit into
release-v1.20.xfrom
konflux/component-updates/operator-1-20-bundle-component-update-opc-1-20-opc
Sep 9, 2026
Merged

chore(deps): update opc-1-20-opc to 64e3b54#29422
pramodbindal merged 1 commit into
release-v1.20.xfrom
konflux/component-updates/operator-1-20-bundle-component-update-opc-1-20-opc

Conversation

@red-hat-konflux-kflux-prd-rh02

Copy link
Copy Markdown
Contributor

Image created from 'https://github.com/openshift-pipelines/p12n-opc.git?rev=06fc308d4d329197f30aac7c60c64f9bdccd5f8e'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/tekton-ecosystem-tenant/pipelines-opc-rhel9 digest 647258d -> 64e3b54

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

Image created from 'https://github.com/openshift-pipelines/p12n-opc.git?rev=06fc308d4d329197f30aac7c60c64f9bdccd5f8e'

Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot added approved Label for Approved PRs automated For Automated PRs konflux For Konflux PRs konflux-nudge lgtm Looks Good to Me Label labels Sep 5, 2026
@pramodbindal
pramodbindal merged commit a8ef360 into release-v1.20.x Sep 9, 2026
@pramodbindal
pramodbindal deleted the konflux/component-updates/operator-1-20-bundle-component-update-opc-1-20-opc branch September 9, 2026 06:29
@openshift-ci

openshift-ci Bot commented Sep 9, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: pramodbindal, red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Label for Approved PRs automated For Automated PRs konflux For Konflux PRs konflux-nudge lgtm Looks Good to Me Label

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant