Skip to content

[bot:release-v1.20.x] Update catalog - #29272

Open
github-actions[bot] wants to merge 1 commit into
release-v1.20.xfrom
actions/update/operator-update-catalog-release-v1.20.x
Open

github-actions[bot] wants to merge 1 commit into
release-v1.20.xfrom
actions/update/operator-update-catalog-release-v1.20.x

Conversation

@github-actions

@github-actions github-actions Bot commented Sep 3, 2026

Copy link
Copy Markdown

No description provided.

@github-actions github-actions Bot added approved Label for Approved PRs lgtm Looks Good to Me Label automated For Automated PRs labels Sep 3, 2026
@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch from 602d51b to b9baa53 Compare September 4, 2026 05:21
@openshift-ci openshift-ci Bot removed the lgtm Looks Good to Me Label label Sep 4, 2026
@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch 5 times, most recently from 02405ee to 0d0feb0 Compare September 9, 2026 05:37
@pramodbindal

Copy link
Copy Markdown
Member

/retest

1 similar comment
@pramodbindal

Copy link
Copy Markdown
Member

/retest

@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch 3 times, most recently from 7d850ee to 4e25be4 Compare September 12, 2026 05:19
@openshift-ci

openshift-ci Bot commented Sep 12, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: github-actions[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch 8 times, most recently from e3e48e4 to ba5d00b Compare September 20, 2026 05:38
@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch 3 times, most recently from 36c3493 to 9135ba5 Compare September 23, 2026 05:29
@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch 6 times, most recently from 8b57687 to 5da248b Compare September 29, 2026 06:29
@github-actions
github-actions Bot force-pushed the actions/update/operator-update-catalog-release-v1.20.x branch from 5da248b to 70b9aac Compare September 30, 2026 06:14
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Label for Approved PRs automated For Automated PRs

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant