SRVKP-10415 set API_URL env variable - #1350
anwesha-palit-redhat merged 1 commit into
Conversation
PR Reviewer Guide 🔍Warning
Here are some key observations to aid the review process:
|
|
Important The |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: anwesha-palit-redhat, bharatmajjiga The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
dd9c06e
into
openshift-pipelines:main
|
/cherry-pick master |
|
@bharatmajjiga: new pull request created: #1353 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Migration - For playwright tests script, added API_URL env variable
Summary
Screen Recordings / Screenshot