Skip to content

docs: text scale meta and hydration - #5422

Merged
eirikbacker merged 3 commits into
mainfrom
docs/text-scale-meta
Sep 24, 2026
Merged

eirikbacker merged 3 commits into
mainfrom
docs/text-scale-meta

Conversation

@eirikbacker

Copy link
Copy Markdown
Contributor

@changeset-bot

changeset-bot Bot commented Sep 24, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f3fb368

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 0 packages

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Sep 24, 2026

Copy link
Copy Markdown
Contributor

Preview deployments for this pull request:

storybook - 24. Sep 2026 - 09:58

themebuilder - 24. Sep 2026 - 09:58

www - 24. Sep 2026 - 10:00

Signed-off-by: Eirik Backer <eirik.backer@gmail.com>
@eirikbacker
eirikbacker merged commit 491d92a into main Sep 24, 2026
14 checks passed
@eirikbacker
eirikbacker deleted the docs/text-scale-meta branch September 24, 2026 09:37
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Recommend <meta name="text-scale" content="scale">?

2 participants