feat(css): postcss-parse-data-color - #5394
Draft
eirikbacker wants to merge 9 commits into
Draft
eirikbacker wants to merge 9 commits into
eirikbacker wants to merge 9 commits into
Conversation
🦋 Changeset detectedLatest commit: 043ad46 The changes in this PR will be included in the next version bump. This PR includes changesets to release 0 packagesWhen changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Contributor
|
Preview deployments for this pull request: storybook - themebuilder - |
Signed-off-by: Michael Marszalek <mimarz@gmail.com>
This branch has not been deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
--ds-props--ds-color-*--(light|dark)automatically to[data-color="SCALE"]@media(max-width: 400px) { [data-color="danger"] { --ds-color: tomato } }is now possible, but since Figma can not replicate this, we now keep the augmenting to top level only :)