Skip to content

fix: Link to intermediate changelog - #5

Open
adamdmharvey wants to merge 1 commit into
backstage:mainfrom
adamdmharvey:patch-1
Open

fix: Link to intermediate changelog#5
adamdmharvey wants to merge 1 commit into
backstage:mainfrom
adamdmharvey:patch-1

Conversation

@adamdmharvey

Copy link
Copy Markdown
Member

Whenever the Backstage service user updates the Version Updates PR, the description has a message which is unusable.

I've tried clicking it before, as have colleagues of mine, just to be redirected to a 404. That's because the message links to an intermediate changelog which doesn't yet actually exist, and the message link is hardcoded to the master branch.

This updates the action to make this link use the intermediate branch name so that users can preview the version updates in the pending changelog before the actual release is cut.

image

The link in the description codes to:
https://github.com/backstage/backstage/blob/master/docs/releases/v1.15.0-next.1-changelog.md

While it SHOULD be:

https://github.com/backstage/backstage/blob/changeset-release/master/docs/releases/v1.15.0-next.1-changelog.md
(which does work and let's me preview the next changelog)

@adamdmharvey

Copy link
Copy Markdown
Member Author

FYI @Rugvip @freben @benjdlambert there's no CODEOWNERS in this repo and I'm not able to tag the maintainers team for some reason. (may not directly have permissions to this repo?)

@adamdmharvey

adamdmharvey commented May 25, 2023

Copy link
Copy Markdown
Member Author

I guess the only question in my mind is before merge/after merge mechanics, in that if I look at this older PR:

backstage/backstage#17831

image

Then the link NOW works, and with my change I have a feeling it wouldn't? Unless this action rewrites it just before merge, which it may not.

So maybe this is an experiment and welcome pros/cons/suggestions!

Potentially we could even add BOTH links, one to the intermediate branch and one to master and sort of parenthetically note which one might work when. sort of covers both use cases - bit brute force but maybe more user friendly?

@benjdlambert

Copy link
Copy Markdown
Member

@adamdmharvey another suggestion would be to use the commit SHA instead of the latest commit, rather than using the branch name which would always be changeset-release/master.

1 similar comment
@benjdlambert

Copy link
Copy Markdown
Member

@adamdmharvey another suggestion would be to use the commit SHA instead of the latest commit, rather than using the branch name which would always be changeset-release/master.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants